Skip to content

cleanup assisted installer bare metal validations enhancement - #393

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
dhellmann:assisted-installer-validations-2
Sep 2, 2020
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
dhellmann:assisted-installer-validations-2

Conversation

@dhellmann

Copy link
Copy Markdown
Contributor

fix some review comments missed before #388 merged

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jun 29, 2020
@dhellmann

Copy link
Copy Markdown
Contributor Author

/cc @hardys @romfreiman I missed a few of your comments on #388 before I removed the hold and let it merge

@dhellmann

Copy link
Copy Markdown
Contributor Author

/assign @hardys

@russellb russellb left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@dhellmann the updates look trivial, but I see it now has a conflict. Once you rebase this, ping me and I'll merge it.

fix some review comments missed before openshift#388 merged

Signed-off-by: Doug Hellmann <dhellmann@redhat.com>
@dhellmann
dhellmann force-pushed the assisted-installer-validations-2 branch from 4c4d803 to a2d893f Compare September 1, 2020 13:09
@dhellmann

Copy link
Copy Markdown
Contributor Author

@russellb this has been rebased

@russellb russellb left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 2, 2020
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dhellmann, russellb

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 8aa9521 into openshift:master Sep 2, 2020
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants