Skip to content

Bug 1784298: Query Browser: Fix "reduced resolution" showing unnecessarily - #7654

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
kyoto:monitoring-fix-reduced-resolution-message
Jan 6, 2021
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
kyoto:monitoring-fix-reduced-resolution-message

Conversation

@kyoto

@kyoto kyoto commented Dec 24, 2020

Copy link
Copy Markdown
Member

The way that samples was being updated sometimes caused a non-optimal
value to be used and could also cause the "Displaying with reduced
resolution due to large dataset" message to be displayed. This fixes the
issue by always setting samples to the preferred value whenever the
graph's timespan changes.

The way that `samples` was being updated sometimes caused a non-optimal
value to be used and could also cause the "Displaying with reduced
resolution due to large dataset" message to be displayed. This fixes the
issue by always setting `samples` to the preferred value whenever the
graph's timespan changes.
@openshift-ci-robot openshift-ci-robot added the bugzilla/severity-low Referenced Bugzilla bug's severity is low for the branch this PR is targeting. label Dec 24, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@kyoto: This pull request references Bugzilla bug 1784298, which is valid. The bug has been moved to the POST state. The bug has been updated to refer to the pull request using the external bug tracker.

3 validation(s) were run on this bug
  • bug is open, matching expected state (open)
  • bug target release (4.7.0) matches configured target release for branch (4.7.0)
  • bug is in the state ASSIGNED, which is one of the valid states (NEW, ASSIGNED, ON_DEV, POST, POST)
Details

In response to this:

Bug 1784298: Query Browser: Fix "reduced resolution" showing unnecessarily

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. label Dec 24, 2020
@openshift-ci-robot openshift-ci-robot added component/core Related to console core functionality approved Indicates a PR has been approved by an approver from all required OWNERS files. component/monitoring Related to monitoring labels Dec 24, 2020
@kyoto

kyoto commented Dec 24, 2020

Copy link
Copy Markdown
Member Author

/retest

@spadgett spadgett left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Jan 5, 2021
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: kyoto, spadgett

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@spadgett

spadgett commented Jan 5, 2021

Copy link
Copy Markdown
Member

The analyze failure should be fixed by #7702

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

7 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-merge-robot
openshift-merge-robot merged commit 56570ab into openshift:master Jan 6, 2021
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@kyoto: All pull requests linked via external trackers have merged:

Bugzilla bug 1784298 has been moved to the MODIFIED state.

Details

In response to this:

Bug 1784298: Query Browser: Fix "reduced resolution" showing unnecessarily

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@kyoto
kyoto deleted the monitoring-fix-reduced-resolution-message branch January 6, 2021 05:24
@spadgett spadgett added this to the v4.7 milestone Jan 7, 2021
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. bugzilla/severity-low Referenced Bugzilla bug's severity is low for the branch this PR is targeting. bugzilla/valid-bug Indicates that a referenced Bugzilla bug is valid for the branch this PR is targeting. component/core Related to console core functionality component/monitoring Related to monitoring lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants