Repository navigation
Bug 1976307: add ImageStreams manifest delete annotation logic - #618
Conversation
|
@jottofar: This pull request references Bugzilla bug 1976307, which is invalid:
Comment DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
/bugzilla refresh |
|
@jottofar: This pull request references Bugzilla bug 1976307, which is valid. The bug has been moved to the POST state. The bug has been updated to refer to the pull request using the external bug tracker. 3 validation(s) were run on this bug
Requesting review from QA contact: DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
This PR is needed by |
|
/retest |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: jottofar, wking The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
LalatenduMohanty
left a comment
There was a problem hiding this comment.
Can we add an unit test for this?
There are no unit tests specific to resource type, e.g. imagestreams vs. deployments. There's only the generic delete annotation unit tests in |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
Actually, thinking about this more, you're probably asking "can we add a unit test that would've caught this issue". Probably so but I don't think there's anything there now to "build on" but I'll look again to see what can be done. |
I looked at this when I originally added the |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
3 similar comments
|
/retest Please review the full test history for this PR and help us cut down flakes. |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
|
/retest Please review the full test history for this PR and help us cut down flakes. |
|
@jottofar: All pull requests linked via external trackers have merged: Bugzilla bug 1976307 has been moved to the MODIFIED state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
No description provided.