Skip to content

pkg/cvo/upgradeable: Fix "Upgradebale" -> "Upgradeable" - #459

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:upgradebale-typo
Feb 10, 2021
Merged

pkg/cvo/upgradeable: Fix "Upgradebale" -> "Upgradeable"#459
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:upgradebale-typo

Conversation

@wking

@wking wking commented Sep 18, 2020

Copy link
Copy Markdown
Member

Fixing a typo from 0452814 (#243).

Generated with:

$ sed -i 's/Upgradebale/Upgradeable/g' $(git grep -l Upgradebale)

No rush; we can land this after 4.6 forks off and master reopens.

Fixing a typo from 0452814 (cvo: long running upgradeable sync
worker, 2019-09-05, openshift#243).

Generated with:

  $ sed -i 's/Upgradebale/Upgradeable/g' $(git grep -l Upgradebale)
@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 18, 2020

@LalatenduMohanty LalatenduMohanty left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Sep 21, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: LalatenduMohanty, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [LalatenduMohanty,wking]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

12 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@sdodson

sdodson commented Sep 22, 2020

Copy link
Copy Markdown
Member

/hold
We need to look at why these tests are always failing

@openshift-ci-robot openshift-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 22, 2020
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wking: The following test failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/prow/e2e 146defe link /test e2e

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@LalatenduMohanty

Copy link
Copy Markdown
Member

/retest

@openshift-merge-robot

Copy link
Copy Markdown
Contributor

@wking: The following tests failed, say /retest to rerun all failed tests:

Test name Commit Details Rerun command
ci/prow/e2e-agnostic-upgrade 146defe link /test e2e-agnostic-upgrade
ci/prow/e2e-agnostic-operator 146defe link /test e2e-agnostic-operator

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@jottofar

jottofar commented Feb 3, 2021

Copy link
Copy Markdown
Contributor

/test e2e-agnostic-upgrade

@wking

wking commented Feb 9, 2021

Copy link
Copy Markdown
Member Author

/retest

@wking

wking commented Feb 10, 2021

Copy link
Copy Markdown
Member Author

e2e-operator passed:

2021/02/09 21:24:27 Container test in pod e2e-agnostic-operator-e2e-test completed successfully

And then died on some gather flake.

/override ci/prow/e2e-agnostic-operator

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@wking: Overrode contexts on behalf of wking: ci/prow/e2e-agnostic-operator

Details

In response to this:

e2e-operator passed:

2021/02/09 21:24:27 Container test in pod e2e-agnostic-operator-e2e-test completed successfully

And then died on some gather flake.

/override ci/prow/e2e-agnostic-operator

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@wking

wking commented Feb 10, 2021

Copy link
Copy Markdown
Member Author

/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Feb 10, 2021
@wking

wking commented Feb 10, 2021

Copy link
Copy Markdown
Member Author

/test unit

The usual race

@openshift-merge-robot
openshift-merge-robot merged commit a9a80fe into openshift:master Feb 10, 2021
@wking
wking deleted the upgradebale-typo branch April 6, 2021 22:22
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants