Skip to content

Add CRD openapi definitions - #168

Merged
openshift-merge-robot merged 4 commits into
openshift:masterfrom
soltysh:crd-validation
Apr 30, 2019
Merged

openshift-merge-robot merged 4 commits into
openshift:masterfrom
soltysh:crd-validation

Conversation

@soltysh

@soltysh soltysh commented Apr 19, 2019

Copy link
Copy Markdown
Contributor

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

@soltysh: GitHub didn't allow me to assign the following users: sttts.

Note that only openshift members and repo collaborators can be assigned and that issues/PRs can only have 10 assignees at the same time.
For more information please see the contributor guide

Details

In response to this:

Fixes https://bugzilla.redhat.com/show_bug.cgi?id=1688638

/assign @deads2k @smarterclayton @sttts

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-ci-robot openshift-ci-robot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Apr 19, 2019
@soltysh

soltysh commented Apr 19, 2019

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@soltysh

soltysh commented Apr 19, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@soltysh

soltysh commented Apr 19, 2019

Copy link
Copy Markdown
Contributor Author

/retest

Comment thread Gopkg.toml

[[prune.project]]
name = "github.com/openshift/library-go"
non-go = false

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

update the commit message with reason for prune exception

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Done.

@soltysh

soltysh commented Apr 23, 2019

Copy link
Copy Markdown
Contributor Author

@abhinavdahiya ptal

@soltysh

soltysh commented Apr 24, 2019

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@soltysh

soltysh commented Apr 24, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@soltysh

soltysh commented Apr 24, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@abhinavdahiya

Copy link
Copy Markdown
Contributor

we can't move forawrd on this as a lot of operators are failing the validation to update the clusteroperator:
example monitoring:

error occurred while setting status to in progress: ClusterOperator.config.openshift.io "monitoring" is invalid: []: Invalid value: map[string]interface {}{"kind":"ClusterOperator", "metadata":map[string]interface {}{"uid":"96475849-66d6-11e9-a6c8-12da682b8aa6", "generation":1, "creationTimestamp":"2019-04-24T21:19:04Z", "resourceVersion":"8619", "name":"monitoring"}, "status":map[string]interface {}{"versions":interface {}(nil), "relatedObjects":interface {}(nil), "extension":interface {}(nil), "conditions":[]interface {}{map[string]interface {}{"lastTransitionTime":"2019-04-24T21:19:04Z", "type":"Failing", "status":"Unknown"}, map[string]interface {}{"type":"Available", "status":"Unknown", "lastTransitionTime":"2019-04-24T21:19:04Z"}, map[string]interface {}{"type":"Progressing", "status":"True", "lastTransitionTime":"2019-04-24T21:19:04Z", "message":"Rolling out the stack."}}}, "spec":map[string]interface {}{}, "apiVersion":"config.openshift.io/v1"}: validation failure list:
status.relatedObjects in body must be of type array: "null"
status.versions in body must be of type array: "null"

@soltysh

soltysh commented Apr 25, 2019

Copy link
Copy Markdown
Contributor Author

Monitoring is using old version of openshift/api which is missing the necessary tags, I've opened openshift/cluster-monitoring-operator#338 to address that.

@soltysh

soltysh commented Apr 25, 2019

Copy link
Copy Markdown
Contributor Author

Actually the PR to look for is openshift/cluster-monitoring-operator#333 - merged

@soltysh

soltysh commented Apr 25, 2019

Copy link
Copy Markdown
Contributor Author

I'm checking other pods/operator for errors as such.

@soltysh

soltysh commented Apr 25, 2019

Copy link
Copy Markdown
Contributor Author

Another to blame is cluster-storage-operator, fix in openshift/cluster-storage-operator#29 - merged

@soltysh

soltysh commented Apr 25, 2019

Copy link
Copy Markdown
Contributor Author

Also operator-marketplace, fix in operator-framework/operator-marketplace#172 - merged

@soltysh

soltysh commented Apr 26, 2019

Copy link
Copy Markdown
Contributor Author

Prereqs merged, let's
/retest

soltysh added 4 commits April 26, 2019 22:42
Adding openshift/library-go requires bumping k8s.io/apiextensions-apiserver
and setting sigs.k8s.io/controller-tools override to point to
openshift's fork. Additionally I'm setting prune.project for library-go
so that this project does not get removed by dep. We are not
explicitly using library-go inside code, other than the generation part
in Makefile.
@soltysh

soltysh commented Apr 26, 2019

Copy link
Copy Markdown
Contributor Author

Rebased, let's see the current results.

@soltysh

soltysh commented Apr 27, 2019

Copy link
Copy Markdown
Contributor Author

/retest

1 similar comment
@soltysh

soltysh commented Apr 29, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@soltysh

soltysh commented Apr 29, 2019

Copy link
Copy Markdown
Contributor Author

Further failures:

@soltysh

soltysh commented Apr 30, 2019

Copy link
Copy Markdown
Contributor Author

/retest

2 similar comments
@soltysh

soltysh commented Apr 30, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@soltysh

soltysh commented Apr 30, 2019

Copy link
Copy Markdown
Contributor Author

/retest

@abhinavdahiya

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Apr 30, 2019
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: abhinavdahiya, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 30, 2019
@openshift-merge-robot
openshift-merge-robot merged commit 23f7ee1 into openshift:master Apr 30, 2019
@soltysh
soltysh deleted the crd-validation branch April 30, 2019 18:59
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants