OCPBUGS-42880: Upgradeable=False should not block a 4.(y+1).z to 4.(y+1).z' retarget - #1094
Conversation
|
@hongkailiu: This pull request references Jira Issue OCPBUGS-42880, which is invalid:
Comment The bug has been updated to refer to the pull request using the external bug tracker. DetailsIn response to this: Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
0cdab63 to
fce734d
Compare
|
/jira refresh |
|
@hongkailiu: This pull request references Jira Issue OCPBUGS-42880, which is valid. The bug has been moved to the POST state. 3 validation(s) were run on this bug
Requesting review from QA contact: DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
9eb95d3 to
efc4c61
Compare
|
/cc |
|
/retest-required |
1 similar comment
|
/retest-required |
|
/test e2e-aws-ovn-techpreview |
73aa5f9 to
8cf5d7a
Compare
|
confirming a reproduction with 4.17.0 -> 4.18.0-ec.1 -> 4.18.0-ec.2. ### cluster bot: launch 4.17.0 aws
$ oc get clusterversion version
NAME VERSION AVAILABLE PROGRESSING SINCE STATUS
version 4.17.0 True False 4m28s Cluster version is 4.17.0
$ oc adm upgrade channel candidate-4.18
$ oc adm upgrade --to 4.18.0-ec.1
### Wait until the update has been accepted ...
oc adm upgrade
info: An upgrade is in progress. Working towards 4.18.0-ec.1: 110 of 901 done (12% complete), waiting on etcd, kube-apiserver
...
$ oc -n openshift-config-managed patch configmap admin-gates --type json -p '[ {"op": "add", "path": "/data/ack-4.17-testing", value: "testing"}]'
$ oc get -n openshift-config-managed configmap admin-gates -o yaml | yq .data -y
ack-4.17-testing: testing
$ oc adm upgrade
info: An upgrade is in progress. Working towards 4.18.0-ec.1: 110 of 901 done (12% complete), waiting on etcd, kube-apiserver
Upgradeable=False
Reason: AdminAckRequired
Message: testing
$ oc adm upgrade --to 4.18.0-ec.2 --allow-upgrade-with-warnings
### it is blocked
$ oc adm upgrade
info: An upgrade is in progress. Working towards 4.18.0-ec.1: 110 of 901 done (12% complete), waiting on etcd, kube-apiserver
Upgradeable=False
Reason: AdminAckRequired
Message: testing
ReleaseAccepted=False
Reason: PreconditionChecks
Message: Preconditions failed for payload loaded version="4.18.0-ec.2" image="quay.io/openshift-release-dev/ocp-release@sha256:a3233d18821af30610bc16d4a3d0f45a5f6bcfbb50eda139b8b82b824a78559f": Precondition "ClusterVersionUpgradeable" failed because of "AdminAckRequired": testing
Upstream: https://api.integration.openshift.com/api/upgrades_info/graph
Channel: candidate-4.18 (available channels: candidate-4.18)
Recommended updates:
VERSION IMAGE
4.18.0-ec.2 quay.io/openshift-release-dev/ocp-release@sha256:a3233d18821af30610bc16d4a3d0f45a5f6bcfbb50eda139b8b82b824a78559f
|
|
With a new 4.17.1 cluster (logs), poking at the rollback user experience: $ oc adm upgrade --allow-explicit-upgrade --force --to-image registry.build09.ci.openshift.org/ci-ln-jpktmvt/release@sha256:bb98538f1fc1fd9deee7efe3397a6a70ddee80b298eb6c0077795fd3eb63cf34
warning: The requested upgrade image is not one of the available updates. You have used --allow-explicit-upgrade for the update to proceed anyway
warning: --force overrides cluster verification of your supplied release image and waives any update precondition failures.
Requested update to release image registry.build09.ci.openshift.org/ci-ln-jpktmvt/release@sha256:bb98538f1fc1fd9deee7efe3397a6a70ddee80b298eb6c0077795fd3eb63cf34And once that's going: $ oc -n openshift-config-managed patch configmap admin-gates --type json -p '[{"op": "add", "path": "/data", "value": {"ack-4.17-testing": "testing"}}]'And once that's going, ask to go back to 4.17.1: $ registry.build09.ci.openshift.org/ci-ln-38qy8yk/release@sha256:e16ac60ac6971e5b6f89c1d818f5ae711c0d63ad6a6a26ffe795c738e8cc4dde
$ oc adm upgrade --allow-explicit-upgrade --allow-upgrade-with-warnings --to-image registry.build09.ci.openshift.org/ci-ln-38qy8yk/release@sha256:e16ac60ac6971e5b6f89c1d818f5ae711c0d63ad6a6a26f
fe795c738e8cc4dde
warning: The requested upgrade image is not one of the available updates. You have used --allow-explicit-upgrade for the update to proceed anyway
warning: --allow-upgrade-with-warnings is bypassing: the cluster is already upgrading:
Reason: ClusterOperatorsUpdating
Message: Working towards 4.18.0-0.test-2024-10-11-181029-ci-ln-jpktmvt-latest: 110 of 890 done (12% complete), waiting on etcd, kube-apiserver
Requested update to release image registry.build09.ci.openshift.org/ci-ln-38qy8yk/release@sha256:e16ac60ac6971e5b6f89c1d818f5ae711c0d63ad6a6a26ffe795c738e8cc4ddeWhich the Rollback precondition rejects, but, new with this pull, the Upgrade condition is ok with: $ oc get -o json clusterversion version | jq -r '.status.conditions[] | select(.type == "ReleaseAccepted").message'
Preconditions failed for payload loaded version="4.17.1" image="registry.build09.ci.openshift.org/ci-ln-38qy8yk/release@sha256:e16ac60ac6971e5b6f89c1d818f5ae711c0d63ad6a6a26ffe795c738e8cc4dde": Multiple precondition checks failed:
* Precondition "ClusterVersionRollback" failed because of "LowDesiredVersion": 4.17.1 is less than the current target 4.18.0-0.test-2024-10-11-181029-ci-ln-jpktmvt-latest and matches the cluster's previous version, but rollbacks that change major or minor versions are not recommended
* Precondition "ClusterVersionRecommendedUpdate" failed because of "UnknownUpdate": RetrievedUpdates=False (VersionNotFound), so the recommended status of updating from 4.18.0-0.test-2024-10-11-181029-ci-ln-jpktmvt-latest to 4.17.1 is unknown.Give up on that rollback request: $ oc adm upgrade --clear
Cancelled requested upgrade to registry.build09.ci.openshift.org/ci-ln-38qy8yk/release@sha256:e16ac60ac6971e5b6f89c1d818f5ae711c0d63ad6a6a26ffe795c738e8cc4ddeAsk to downgrade to an earlier 4.18: $ oc adm release info registry.ci.openshift.org/ocp/release:4.18.0-0.nightly-2024-10-10-195326 | head -n3
Name: 4.18.0-0.nightly-2024-10-10-195326
Digest: sha256:dbbaaf5235c90c4aafb24292da4d4e22685108bd01f70005703bd17576489d86
Created: 2024-10-10T19:55:13Z
$ oc adm upgrade --allow-explicit-upgrade --allow-upgrade-with-warnings --to-image registry.ci.openshift.org/ocp/release@sha256:dbbaaf5235c90c4aafb24292da4d4e22685108bd01f70005703bd17576489d86
warning: The requested upgrade image is not one of the available updates. You have used --allow-explicit-upgrade for the update to proceed anyway
warning: --allow-upgrade-with-warnings is bypassing: the cluster is already upgrading:
Reason: ClusterOperatorsUpdating
Message: Working towards 4.18.0-0.test-2024-10-11-181029-ci-ln-jpktmvt-latest: 726 of 890 done (81% complete), waiting on dns, network
Requested update to release image registry.ci.openshift.org/ocp/release@sha256:dbbaaf5235c90c4aafb24292da4d4e22685108bd01f70005703bd17576489d86And it's also rejected, but on the earlier signature-check, so we will need to wait until we have an Engineering Candidate with this pull request in it before we can test $ oc get -o json clusterversion version | jq -r '.status.conditions[] | select(.type == "ReleaseAccepted").message'
Retrieving payload failed version="" image="registry.ci.openshift.org/ocp/release@sha256:dbbaaf5235c90c4aafb24292da4d4e22685108bd01f70005703bd17576489d86" failure=The update cannot be verified: unable to verify sha256:dbbaaf5235c90c4aafb24292da4d4e22685108bd01f70005703bd17576489d86 against keyrings: verifier-public-key-redhat |
| clusterVersion.Status.Conditions = append(clusterVersion.Status.Conditions, configv1.ClusterOperatorStatusCondition{ | ||
| Type: configv1.OperatorUpgradeable, | ||
| Status: *tc.upgradeable, | ||
| Reason: "bla", |
|
/cherrypick release-4.17 |
|
@wking: once the present PR merges, I will cherry-pick it on top of release-4.17 in a new PR and assign it to you. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
petr-muller
left a comment
There was a problem hiding this comment.
LGTM with a nit, feel free to unhold
/hold
Co-authored-by: Petr Muller <afri@afri.cz>
|
/label tide/merge-method-squash |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: hongkailiu, petr-muller, wking The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
@hongkailiu: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
pre-merge verified in https://issues.redhat.com/browse/OCPBUGS-42880?focusedId=25803260&page=com.atlassian.jira.plugin.system.issuetabpanels:comment-tabpanel#comment-25803260 |
|
@hongkailiu: This pull request references Jira Issue OCPBUGS-42880, which is valid. 3 validation(s) were run on this bug
Requesting review from QA contact: DetailsIn response to this: Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
following the findings and retest in OCPBUGS-42880#comment-25883694 the issues are not directly related to the fix. passing as verified. |
|
@hongkailiu: Jira Issue OCPBUGS-42880: All pull requests linked via external trackers have merged: Jira Issue OCPBUGS-42880 has been moved to the MODIFIED state. DetailsIn response to this: Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
@wking: new pull request created: #1095 DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[ART PR BUILD NOTIFIER] Distgit: cluster-version-operator |
https://issues.redhat.com/browse/OCPBUGS-42880