Bug 2001974: delete hello-openshift in payload imagestream via CVO annotation - #392
gabemontero wants to merge 1 commit into
Conversation
|
@gabemontero: This pull request references Bugzilla bug 2001974, which is valid. The bug has been moved to the POST state. The bug has been updated to refer to the pull request using the external bug tracker. 6 validation(s) were run on this bug
Requesting review from QA contact: DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: gabemontero The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/hold I want to confirm |
|
/test unit |
|
samples clean, other failed, e2e-aws-upgrade /retest |
|
the hello openshift imagestream is still present in 4.8 that installer annotation may not be supported in 4.8 ... investigating |
|
Confirmed it has not been backported. Have inquiry to see if it will be. Inquiring int https://bugzilla.redhat.com/show_bug.cgi?id=1976307 to see if it will be. |
|
@gabemontero: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here. |
|
/skip |
|
per @jottofar the annotation is not getting backported just closing this out |
|
@gabemontero: This pull request references Bugzilla bug 2001974. The bug has been updated to no longer refer to the pull request using the external bug tracker. All external bug links have been closed. The bug has been moved to the NEW state. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
cherry pick failed ... example #380 (comment)
ultimately want to get this back to 4.7 .... see #385 (comment)
/assign @dperaza4dustbit
more tech debt I was still carrying prior to hand off