Skip to content

don't manage apiservices until daemonset is up - #30

Merged
openshift-merge-robot merged 2 commits into
openshift:masterfrom
deads2k:apiservices
Oct 8, 2018
Merged

don't manage apiservices until daemonset is up#30
openshift-merge-robot merged 2 commits into
openshift:masterfrom
deads2k:apiservices

Conversation

@deads2k

@deads2k deads2k commented Oct 8, 2018

Copy link
Copy Markdown
Contributor

This prevents the initial creation of apiservices that won't work.

/assign @mfojtik

@openshift-ci-robot openshift-ci-robot added size/M Denotes a PR that changes 30-99 lines, ignoring generated files. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Oct 8, 2018
@mfojtik

mfojtik commented Oct 8, 2018

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Oct 8, 2018
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: deads2k, mfojtik

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 2226a01 into openshift:master Oct 8, 2018
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants