Skip to content

Consume cluster ID from config.openshift.io - #88

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
ironcladlou:clusterid
Dec 17, 2018
Merged

Consume cluster ID from config.openshift.io#88
openshift-merge-robot merged 1 commit into
openshift:masterfrom
ironcladlou:clusterid

Conversation

@ironcladlou

Copy link
Copy Markdown
Contributor

Consume cluster ID from its stable location in config.openshift.io. The previous
field on the install configmap is deprecated.

Blocks openshift/installer#783.

@openshift-ci-robot openshift-ci-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Dec 14, 2018
@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 14, 2018
@Miciah

Miciah commented Dec 14, 2018

Copy link
Copy Markdown
Contributor

Would be good to update the reference to the 'openshiftClusterID' tag in

// ClusterID is the value of the 'openshiftClusterID' tag identifying the
but otherwise Edit: That's referring to the AWS tag name.
/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Dec 14, 2018
@Miciah

Miciah commented Dec 14, 2018

Copy link
Copy Markdown
Contributor
The maximum number of VPC endpoints has been reached

/retest

@Miciah

Miciah commented Dec 14, 2018

Copy link
Copy Markdown
Contributor

e2e-aws and e2e-aws-operator are both failing because of the VPC limit.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

5 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@Miciah

Miciah commented Dec 15, 2018

Copy link
Copy Markdown
Contributor
time="2018-12-15T20:42:59Z" level=fatal msg="failed to get clusterversion 'version': clusterversions.config.openshift.io \"version\" is forbidden: User \"system:serviceaccount:openshift-ingress-operator:ingress-operator\" cannot get clusterversions.config.openshift.io at the cluster scope: no RBAC policy matched"

@ironcladlou, you need to add "clusterversions" to the list of resources here:

- apiGroups:
- config.openshift.io
resources:
- ingresses
verbs:
- get

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

10 similar comments
@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

@openshift-bot

Copy link
Copy Markdown
Contributor

/retest

Please review the full test history for this PR and help us cut down flakes.

Consume cluster ID from its stable location in config.openshift.io. The previous
field on the install configmap is deprecated.
@openshift-ci-robot openshift-ci-robot removed the lgtm Indicates that a PR is ready to be merged. label Dec 17, 2018
@knobunc

knobunc commented Dec 17, 2018

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Dec 17, 2018
@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ironcladlou, knobunc, Miciah

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [ironcladlou,knobunc]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 2cd0ce3 into openshift:master Dec 17, 2018
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants