Skip to content

pkg/controllers: Use library-go's IngressURI helper - #358

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:library-go-route-ingress-uri
Oct 12, 2020
Merged

pkg/controllers: Use library-go's IngressURI helper#358
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:library-go-route-ingress-uri

Conversation

@wking

@wking wking commented Oct 6, 2020

Copy link
Copy Markdown
Member

Using the helper from openshift/library-go#911 to centralize some route handling and error message generation.

Generated with:

$ emacs pkg/... # everything in the diff
$ go get github.com/openshift/library-go@f360b9835cc8815eccc26cb52dafe05d3cbb3e93
go: github.com/openshift/library-go f360b9835cc8815eccc26cb52dafe05d3cbb3e93 => v0.0.0-20201006230840-f360b9835cc8
$ go mod tidy
$ go mod vendor
$ git add -A go.* vendor

using:

$ go version
go version go1.15.2 linux/amd64

Using the helper from [1] to centralize some route handling and error
message generation.

Generated with:

  $ emacs pkg/... test/library/waits.go # everything in the diff
  $ go get github.com/openshift/library-go@f360b9835cc8815eccc26cb52dafe05d3cbb3e93
  go: github.com/openshift/library-go f360b9835cc8815eccc26cb52dafe05d3cbb3e93 => v0.0.0-20201006230840-f360b9835cc8
  $ go mod tidy
  $ go mod vendor
  $ git add -A go.* vendor

using:

  $ go version
  go version go1.15.2 linux/amd64

[1]: openshift/library-go#911
@wking
wking force-pushed the library-go-route-ingress-uri branch from c5a4fed to ef28bf0 Compare October 7, 2020 03:06
@wking

wking commented Oct 7, 2020

Copy link
Copy Markdown
Member Author

/assign @sttts

@sttts

sttts commented Oct 12, 2020

Copy link
Copy Markdown
Contributor

/retest
Lgtm
/assign @stlaz

@sttts

sttts commented Oct 12, 2020

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 12, 2020
@mfojtik

mfojtik commented Oct 12, 2020

Copy link
Copy Markdown
Contributor

/lgtm

nice refactoring @wking

@openshift-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mfojtik, sttts, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Oct 12, 2020
@openshift-merge-robot
openshift-merge-robot merged commit b9d7a84 into openshift:master Oct 12, 2020
@wking
wking deleted the library-go-route-ingress-uri branch October 22, 2020 13:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants