Skip to content

Enable 4.5.22 in candidate channel(s) - #570

Merged
openshift-merge-robot merged 1 commit into
masterfrom
pr-candidate-4.5.22
Dec 2, 2020
Merged

Enable 4.5.22 in candidate channel(s)#570
openshift-merge-robot merged 1 commit into
masterfrom
pr-candidate-4.5.22

Conversation

@openshift-bot

Copy link
Copy Markdown
Contributor

Please merge immediately. This PR does not need to wait for an advisory to ship, but the associated advisory is https://errata.devel.redhat.com/advisory/66568 .

This PR will also enable upgrades from 4.5.22 to releases in candidate-4.6

@vrutkovs

vrutkovs commented Dec 2, 2020

Copy link
Copy Markdown

Autoapproving candidate PRs
/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Dec 2, 2020
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: openshift-bot, vrutkovs

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 2, 2020
@openshift-merge-robot
openshift-merge-robot merged commit 89f1c20 into master Dec 2, 2020
@sdodson
sdodson deleted the pr-candidate-4.5.22 branch December 8, 2020 18:48
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants