Skip to content

Enable 4.5.21 in stable channel(s) - #561

Merged
openshift-merge-robot merged 1 commit into
masterfrom
pr-stable-4.5.21
Dec 9, 2020
Merged

Enable 4.5.21 in stable channel(s)#561
openshift-merge-robot merged 1 commit into
masterfrom
pr-stable-4.5.21

Conversation

@openshift-bot

Copy link
Copy Markdown
Contributor

Please merge within 48 hours of https://errata.devel.redhat.com/advisory/66372 shipping live OR a Cincinnati-first release.

This should provide adequate soak time for fast channel PR #560

This PR will also enable upgrades from 4.5.21 to releases in stable-4.6

@openshift-bot openshift-bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Nov 26, 2020
@sdodson

sdodson commented Dec 8, 2020

Copy link
Copy Markdown
Member

/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Dec 8, 2020
@vrutkovs

vrutkovs commented Dec 9, 2020

Copy link
Copy Markdown

Approving stable PRs after 48 hours
/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Dec 9, 2020
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: openshift-bot, vrutkovs

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 9, 2020
@openshift-merge-robot
openshift-merge-robot merged commit 76373b7 into master Dec 9, 2020
@sdodson
sdodson deleted the pr-stable-4.5.21 branch December 14, 2020 17:57
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants