Skip to content

Enable 4.6.1 in stable channel(s) - #519

Merged
openshift-merge-robot merged 1 commit into
masterfrom
pr-stable-4.6.1
Oct 27, 2020
Merged

openshift-merge-robot merged 1 commit into
masterfrom
pr-stable-4.6.1

Conversation

@openshift-bot

Copy link
Copy Markdown
Contributor

Please merge within 48 hours of https://errata.devel.redhat.com/advisory/54579 shipping live OR a Cincinnati-first release.

This should provide adequate soak time for fast channel PR #518

@openshift-bot openshift-bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Oct 22, 2020
Comment thread channels/stable-4.6.yaml Outdated
@sdodson

sdodson commented Oct 27, 2020

Copy link
Copy Markdown
Member

Errata has shipped https://access.redhat.com/errata/RHBA-2020:4196
/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Oct 27, 2020
@wking

wking commented Oct 27, 2020

Copy link
Copy Markdown
Member

Fast landed, and there are no edges in stable, so land this too.

/lgtm

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: openshift-bot, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Oct 27, 2020
@vrutkovs

Copy link
Copy Markdown

/retest

1 similar comment
@wking

wking commented Oct 27, 2020

Copy link
Copy Markdown
Member

/retest

@wking

wking commented Oct 27, 2020

Copy link
Copy Markdown
Member

Hmm, no new jobs. Kick it harder.

/test images
/test publish

@openshift-merge-robot
openshift-merge-robot merged commit f99a471 into master Oct 27, 2020
@sdodson
sdodson deleted the pr-stable-4.6.1 branch December 8, 2020 18:47
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants