Skip to content

internal-channels/stable: Promote 4.10.4 - #1660

Merged
openshift-merge-robot merged 1 commit into
masterfrom
promote-4.10.4-to-stable
Mar 23, 2022
Merged

openshift-merge-robot merged 1 commit into
masterfrom
promote-4.10.4-to-stable

Conversation

@vrutkovs

Copy link
Copy Markdown

It was promoted to the feeder fast by ce7a055 (Merge pull request #1632 from openshift/promote-4.10.4-to-fast, 2022-03-16) 7 days, 0:57:39.771699 ago.

It was promoted to the feeder fast by ce7a055 (Merge pull request #1632
from openshift/promote-4.10.4-to-fast, 2022-03-16) 7 days, 0:57:39.771699
ago.
@vrutkovs vrutkovs added approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. labels Mar 23, 2022
@openshift-ci

openshift-ci Bot commented Mar 23, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: vrutkovs

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@openshift-ci

openshift-ci Bot commented Mar 23, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: vrutkovs

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 37c8bb6 into master Mar 23, 2022
@openshift-ci

openshift-ci Bot commented Mar 23, 2022

Copy link
Copy Markdown
Contributor

@vrutkovs: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

vrutkovs pushed a commit that referenced this pull request Mar 23, 2022
It was promoted to the feeder stable by 37c8bb6 (Merge pull request #1660
from openshift/promote-4.10.4-to-stable, 2022-03-23) 0:47:45.378523 ago.
vrutkovs pushed a commit that referenced this pull request Mar 23, 2022
It was promoted to the feeder stable by 37c8bb6 (Merge pull request #1660
from openshift/promote-4.10.4-to-stable, 2022-03-23) 0:47:57.481349 ago.
@wking
wking deleted the promote-4.10.4-to-stable branch March 30, 2022 22:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants