Skip to content

channels/fast-4.9: Promote 4.9.12 - #1331

Merged
openshift-merge-robot merged 1 commit into
masterfrom
promote-4.9.12-to-fast-4.9
Jan 4, 2022
Merged

openshift-merge-robot merged 1 commit into
masterfrom
promote-4.9.12-to-fast-4.9

Conversation

@vrutkovs

@vrutkovs vrutkovs commented Jan 4, 2022

Copy link
Copy Markdown

It was promoted to the feeder fast by 7fea5cf (Merge pull request #1330 from openshift/promote-4.9.12-to-fast, 2022-01-04) 0:52:26.316609 ago.

It was promoted to the feeder fast by 7fea5cf (Merge pull request #1330
from openshift/promote-4.9.12-to-fast, 2022-01-04) 0:52:26.316609 ago.
@vrutkovs vrutkovs added approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. labels Jan 4, 2022
@openshift-ci

openshift-ci Bot commented Jan 4, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: vrutkovs

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@openshift-ci

openshift-ci Bot commented Jan 4, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: vrutkovs

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Jan 4, 2022

Copy link
Copy Markdown
Contributor

@vrutkovs: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-robot
openshift-merge-robot merged commit d2aaa4e into master Jan 4, 2022
@wking
wking deleted the promote-4.9.12-to-fast-4.9 branch January 13, 2022 06:21
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants