Skip to content

hack/show-edges: Teach to walk internal-channels - #1275

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:teach-show-edges-about-internal-channels
Jan 19, 2022
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:teach-show-edges-about-internal-channels

Conversation

@wking

@wking wking commented Dec 9, 2021

Copy link
Copy Markdown
Member

Because showing graphs for those can be convenient too. Catches up with stabilization-changes pivots in bda213e (#112).

Because showing graphs for those can be convenient too.  Catches up
with stabilization-changes pivots in bda213e
(internal-channels/candidate: Add a version-agnostic candidate
channel, 2021-11-03, openshift#112).
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 9, 2021

@LalatenduMohanty LalatenduMohanty left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci

openshift-ci Bot commented Jan 19, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: LalatenduMohanty, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:
  • OWNERS [LalatenduMohanty,wking]

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jan 19, 2022
@openshift-ci

openshift-ci Bot commented Jan 19, 2022

Copy link
Copy Markdown
Contributor

@wking: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-robot
openshift-merge-robot merged commit d4ce983 into openshift:master Jan 19, 2022
@wking
wking deleted the teach-show-edges-about-internal-channels branch January 19, 2022 22:42
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants