Skip to content

Block upgrades to 4.3.5 and 4.2.22 - #125

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
marun:unique-service-ca-serial
Mar 18, 2020
Merged

Block upgrades to 4.3.5 and 4.2.22#125
openshift-merge-robot merged 1 commit into
openshift:masterfrom
marun:unique-service-ca-serial

Conversation

@marun

@marun marun commented Mar 18, 2020

Copy link
Copy Markdown
Contributor

These releases enable automated ca rotation without unique ca serial numbers which can break non-golang tls clients as per [1].

1: https://bugzilla.redhat.com/show_bug.cgi?id=1810036

These releases enable automated ca rotation without unique ca serial
numbers which can break non-golang tls clients as per [1].

1: https://bugzilla.redhat.com/show_bug.cgi?id=1810036
@LalatenduMohanty

Copy link
Copy Markdown
Member

/lgtm
If we want to add more information we can do that in a following PR as we need to block the edges asap first.

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Mar 18, 2020
@mfojtik

mfojtik commented Mar 18, 2020

Copy link
Copy Markdown

@jwforres

Copy link
Copy Markdown
Member

/approve

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jwforres, LalatenduMohanty, marun, mfojtik

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 18, 2020
@openshift-merge-robot
openshift-merge-robot merged commit d980578 into openshift:master Mar 18, 2020
@sdodson

sdodson commented Mar 18, 2020

Copy link
Copy Markdown
Member

https://bugzilla.redhat.com/show_bug.cgi?id=1810036#c8 has the summarized impact assessment

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

7 participants