Skip to content

channels/candidate-4.4: Add 4.4.0-rc.0 and its update source 4.3.5 - #114

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:4.4-rc.0
Mar 12, 2020
Merged

channels/candidate-4.4: Add 4.4.0-rc.0 and its update source 4.3.5#114
openshift-merge-robot merged 1 commit into
openshift:masterfrom
wking:4.4-rc.0

Conversation

@wking

@wking wking commented Mar 12, 2020

Copy link
Copy Markdown
Member
$ oc adm release info quay.io/openshift-release-dev/ocp-release:4.4.0-rc.0-x86_64 | grep Upgrades
  Upgrades: 4.3.5

  $ oc adm release info quay.io/openshift-release-dev/ocp-release:4.4.0-rc.0-x86_64 | grep Upgrades
    Upgrades: 4.3.5
@openshift-ci-robot openshift-ci-robot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Mar 12, 2020
@sdodson

sdodson commented Mar 12, 2020

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Mar 12, 2020
@wking

wking commented Mar 12, 2020

Copy link
Copy Markdown
Member Author

From the amd64 release controller page, 4.3.5 -> 4.4.0-rc.0 had two AWS success and three failures:

Only the latter is an update-stability issue, and we're not considering unreachable workloads as update blockers at the moment (certainly not for the candidate channel).

@eparis

eparis commented Mar 12, 2020

Copy link
Copy Markdown
Member

/approve

@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: eparis, sdodson, wking

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 12, 2020
@openshift-merge-robot
openshift-merge-robot merged commit 6d0e5a4 into openshift:master Mar 12, 2020
@wking
wking deleted the 4.4-rc.0 branch March 19, 2020 20:58
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants