Skip to content
This repository was archived by the owner on Jun 14, 2019. It is now read-only.

Change the configuration file format to YAML - #121

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
petr-muller:yaml-all-the-things
Aug 31, 2018
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
petr-muller:yaml-all-the-things

Conversation

@petr-muller

@petr-muller petr-muller commented Aug 31, 2018

Copy link
Copy Markdown
Member

SSIA

JSON is a subset of YAML, so ci-operator will be able to read also old JSON configs. Using https://github.com/ghodss/yaml means the necessary code changes are minimal because it reuses the json struct member annotations plus it's already vendored (it's used for similar purposes at least by k8s.io/test-infra).

Resolves: #46

@openshift-ci-robot openshift-ci-robot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Aug 31, 2018
@petr-muller

Copy link
Copy Markdown
Member Author

/hold

@openshift-ci-robot openshift-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 31, 2018
@petr-muller

Copy link
Copy Markdown
Member Author

/cc @stevekuznetsov

@stevekuznetsov stevekuznetsov left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

That's nice and clean :)
/lgtm

@openshift-ci-robot openshift-ci-robot added lgtm Indicates that a PR is ready to be merged. approved Indicates a PR has been approved by an approver from all required OWNERS files. labels Aug 31, 2018
@stevekuznetsov

Copy link
Copy Markdown
Contributor

/approve
/hold cancel

@openshift-ci-robot openshift-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Aug 31, 2018
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: petr-muller, stevekuznetsov

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 3bce7f6 into openshift:master Aug 31, 2018
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants