Skip to content

Add oVirt platform status - #451

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
rgolangh:ovirt
Oct 10, 2019
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
rgolangh:ovirt

Conversation

@rgolangh

@rgolangh rgolangh commented Sep 27, 2019

Copy link
Copy Markdown
Contributor

Add oVirt platform status

enhancement: openshift/enhancements#61

Signed-off-by: Roy Golan <rgolan@redhat.com>
@openshift-ci-robot openshift-ci-robot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Sep 27, 2019
@deads2k

deads2k commented Sep 27, 2019

Copy link
Copy Markdown
Contributor

Please link the enhancement to this PR so we can understand the impact this has on 4.3

@deads2k

deads2k commented Oct 3, 2019

Copy link
Copy Markdown
Contributor

Please link the enhancement to this PR so we can understand the impact this has on 4.3

Can you link the enhancement please

@sdodson

sdodson commented Oct 10, 2019

Copy link
Copy Markdown
Member

/approve
Seems inline with other platform definitions.

@abhinavdahiya PTAL

@soltysh soltysh left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve
Leaving final call to @abhinavdahiya like Scott pointed out.

@openshift-ci-robot openshift-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 10, 2019
// by components inside the cluster, like kubelets using the infrastructure rather
// than Kubernetes networking. It is the IP that the Infrastructure.status.apiServerInternalURI
// points to. It is the IP for a self-hosted load balancer in front of the API servers.
APIServerInternalIP string `json:"apiServerInternalIP,omitempty"`

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

so these fields are required but omitempty ??

@sttts / @deads2k any guidance here?

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

omitempty with string will cause they are not printed when empty, but it doesn't make any difference during parsing. I don't think we have clear guidance for it, though.

@sttts sttts Oct 10, 2019

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

omitempty for optional fields is often a good idea. Usually only if you want to distinguish undefined and zero value, you don't use omitempty. Though then these fields are also pointers to survive Golang roundtripping.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

https://github.com/openshift/api/blob/master/config/v1/doc.go#L5

all fields are kindof optional at package level, and this is similar to other structs, so following that seems fine to me.

OpenStack *OpenStackPlatformStatus `json:"openstack,omitempty"`

// Ovirt contains settings specific to the oVirt infrastructure provider.
// +optional

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

@damemi one needs the kubebuilder optional tag as well, doesn't one?

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

though it does not matter here as the default is optional already

@abhinavdahiya

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci-robot openshift-ci-robot added the lgtm Indicates that a PR is ready to be merged. label Oct 10, 2019
@openshift-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: abhinavdahiya, rgolangh, sdodson, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-robot
openshift-merge-robot merged commit 881ce6b into openshift:master Oct 10, 2019
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

8 participants