Skip to content

Do not register new ICSP CRDs, yet - #1164

Merged
openshift-merge-robot merged 1 commit into
openshift:masterfrom
soltysh:drop_newicsp
Apr 8, 2022
Merged

openshift-merge-robot merged 1 commit into
openshift:masterfrom
soltysh:drop_newicsp

Conversation

@soltysh

@soltysh soltysh commented Apr 7, 2022 •

Copy link
Copy Markdown
Contributor

This is followup from #1126 to ensure we don't register those new ICSP CRDs, yet.

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 7, 2022
@openshift-ci
openshift-ci Bot requested review from adambkaplan and bparees April 7, 2022 13:50
@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Apr 7, 2022
@rphillips

Copy link
Copy Markdown
Contributor

/lgtm
/hold cancel

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Apr 7, 2022
@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Apr 7, 2022
@openshift-ci

openshift-ci Bot commented Apr 7, 2022

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rphillips, soltysh

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@soltysh

soltysh commented Apr 8, 2022

Copy link
Copy Markdown
Contributor Author

/label px-approved
/label docs-approved
/label qe-approved

@openshift-ci openshift-ci Bot added px-approved Signifies that Product Support has signed off on this PR docs-approved Signifies that Docs has signed off on this PR qe-approved Signifies that QE has signed off on this PR labels Apr 8, 2022
@openshift-ci

openshift-ci Bot commented Apr 8, 2022

Copy link
Copy Markdown
Contributor

@soltysh: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@openshift-merge-robot
openshift-merge-robot merged commit 01a2835 into openshift:master Apr 8, 2022
@soltysh
soltysh deleted the drop_newicsp branch April 8, 2022 16:03
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. docs-approved Signifies that Docs has signed off on this PR lgtm Indicates that a PR is ready to be merged. px-approved Signifies that Product Support has signed off on this PR qe-approved Signifies that QE has signed off on this PR

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants