Skip to content

Add CAPI e2e Makefile target - #1777

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
honza:capi-e2e
Jul 9, 2025
Merged

Add CAPI e2e Makefile target#1777
openshift-merge-bot[bot] merged 1 commit into
openshift-metal3:masterfrom
honza:capi-e2e

Conversation

@honza

@honza honza commented Jul 8, 2025

Copy link
Copy Markdown
Member

No description provided.

@dtantsur

dtantsur commented Jul 8, 2025

Copy link
Copy Markdown
Member

/approve

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 8, 2025
@dtantsur

dtantsur commented Jul 8, 2025

Copy link
Copy Markdown
Member

/approve

CI failures unrelated, probably CentOS not feeling well.

@openshift-ci

openshift-ci Bot commented Jul 8, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dtantsur

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@elfosardo

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 8, 2025
@elfosardo

Copy link
Copy Markdown
Member

/retest

5 similar comments
@elfosardo

Copy link
Copy Markdown
Member

/retest

@elfosardo

Copy link
Copy Markdown
Member

/retest

@elfosardo

Copy link
Copy Markdown
Member

/retest

@elfosardo

Copy link
Copy Markdown
Member

/retest

@elfosardo

Copy link
Copy Markdown
Member

/retest

@openshift-merge-bot
openshift-merge-bot Bot merged commit 5604f90 into openshift-metal3:master Jul 9, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants