NO-JIRA: skipping add-nodes step for ocp versions lower than 4.17 - #1725
Conversation
|
/pj-rehearse |
b860cfb to
6f4b032
Compare
|
/test images |
|
/lgtm |
|
/test ? |
|
@andfasano: The following commands are available to trigger required jobs: The following commands are available to trigger optional jobs: Use DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
/approve |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: bfournie The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
Required by openshift/release#59335