Skip to content

Fix failure if nmstateconfig.yaml doesn't exist - #1552

Merged
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
bfournie:rm-nmstate
Jul 31, 2023
Merged

Fix failure if nmstateconfig.yaml doesn't exist#1552
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
bfournie:rm-nmstate

Conversation

@bfournie

Copy link
Copy Markdown
Contributor

Fix script failure if the file doesn't exist.

Fix script failure if the file doesn't exist.
@openshift-ci
openshift-ci Bot requested review from lranjbar and zaneb July 17, 2023 02:08
@openshift-ci

openshift-ci Bot commented Jul 17, 2023

Copy link
Copy Markdown

@bfournie: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-agent-compact e8ca9ce link true /test e2e-agent-compact

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@andfasano

Copy link
Copy Markdown
Member

/approve

@openshift-ci

openshift-ci Bot commented Jul 31, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: andfasano

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 31, 2023

@pawanpinjarkar pawanpinjarkar left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 31, 2023
@rwsu

rwsu commented Jul 31, 2023

Copy link
Copy Markdown
Contributor

/test e2e-agent-compact

@openshift-ci

openshift-ci Bot commented Jul 31, 2023

Copy link
Copy Markdown

@rwsu: The specified target(s) for /test were not found.
The following commands are available to trigger required jobs:

  • /test e2e-agent-compact-ipv4
  • /test e2e-metal-ipi-bm
  • /test e2e-metal-ipi-ovn-ipv6
  • /test e2e-metal-ipi-serial-ipv4
  • /test images

The following commands are available to trigger optional jobs:

  • /test e2e-agent-ha-dualstack
  • /test e2e-agent-sno-ipv6
  • /test e2e-metal-ipi-bm-bond
  • /test e2e-metal-ipi-ovn-dualstack
  • /test e2e-metal-ipi-serial-ovn-ipv6
  • /test e2e-metal-ipi-upgrade
  • /test e2e-metal-ipi-upgrade-ovn-ipv6
  • /test e2e-metal-ipi-virtualmedia

Use /test all to run the following jobs that were automatically triggered:

  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-compact-ipv4
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-ha-dualstack
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-sno-ipv6
  • pull-ci-openshift-metal3-dev-scripts-master-images
Details

In response to this:

/test e2e-agent-compact

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@pawanpinjarkar

Copy link
Copy Markdown
Contributor

/test ci/prow/e2e-agent-compact

@openshift-ci

openshift-ci Bot commented Jul 31, 2023

Copy link
Copy Markdown

@pawanpinjarkar: The specified target(s) for /test were not found.
The following commands are available to trigger required jobs:

  • /test e2e-agent-compact-ipv4
  • /test e2e-metal-ipi-bm
  • /test e2e-metal-ipi-ovn-ipv6
  • /test e2e-metal-ipi-serial-ipv4
  • /test images

The following commands are available to trigger optional jobs:

  • /test e2e-agent-ha-dualstack
  • /test e2e-agent-sno-ipv6
  • /test e2e-metal-ipi-bm-bond
  • /test e2e-metal-ipi-ovn-dualstack
  • /test e2e-metal-ipi-serial-ovn-ipv6
  • /test e2e-metal-ipi-upgrade
  • /test e2e-metal-ipi-upgrade-ovn-ipv6
  • /test e2e-metal-ipi-virtualmedia

Use /test all to run the following jobs that were automatically triggered:

  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-compact-ipv4
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-ha-dualstack
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-sno-ipv6
  • pull-ci-openshift-metal3-dev-scripts-master-images
Details

In response to this:

/test ci/prow/e2e-agent-compact

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@bfournie

Copy link
Copy Markdown
Contributor Author

/test e2e-agent-compact

@openshift-ci

openshift-ci Bot commented Jul 31, 2023

Copy link
Copy Markdown

@bfournie: The specified target(s) for /test were not found.
The following commands are available to trigger required jobs:

  • /test e2e-agent-compact-ipv4
  • /test e2e-metal-ipi-bm
  • /test e2e-metal-ipi-ovn-ipv6
  • /test e2e-metal-ipi-serial-ipv4
  • /test images

The following commands are available to trigger optional jobs:

  • /test e2e-agent-ha-dualstack
  • /test e2e-agent-sno-ipv6
  • /test e2e-metal-ipi-bm-bond
  • /test e2e-metal-ipi-ovn-dualstack
  • /test e2e-metal-ipi-serial-ovn-ipv6
  • /test e2e-metal-ipi-upgrade
  • /test e2e-metal-ipi-upgrade-ovn-ipv6
  • /test e2e-metal-ipi-virtualmedia

Use /test all to run the following jobs that were automatically triggered:

  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-compact-ipv4
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-ha-dualstack
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-sno-ipv6
  • pull-ci-openshift-metal3-dev-scripts-master-images
Details

In response to this:

/test e2e-agent-compact

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@bfournie

Copy link
Copy Markdown
Contributor Author

/retest-required

@pawanpinjarkar

Copy link
Copy Markdown
Contributor

/override ci/prow/e2e-agent-compact

@openshift-ci

openshift-ci Bot commented Jul 31, 2023

Copy link
Copy Markdown

@pawanpinjarkar: pawanpinjarkar unauthorized: /override is restricted to Repo administrators, approvers in top level OWNERS file, and the following github teams:.

Details

In response to this:

/override ci/prow/e2e-agent-compact

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@bfournie

Copy link
Copy Markdown
Contributor Author

/override ci/prow/e2e-agent-compact

@openshift-ci

openshift-ci Bot commented Jul 31, 2023

Copy link
Copy Markdown

@bfournie: Overrode contexts on behalf of bfournie: ci/prow/e2e-agent-compact

Details

In response to this:

/override ci/prow/e2e-agent-compact

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-merge-robot
openshift-merge-robot merged commit af56270 into openshift-metal3:master Jul 31, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants