Skip to content

OPNET-200: Add support for ipv6-primary dual stack - #1510

Merged
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
cybertron:v6-primary
Mar 6, 2023
Merged

OPNET-200: Add support for ipv6-primary dual stack#1510
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
cybertron:v6-primary

Conversation

@cybertron

Copy link
Copy Markdown
Contributor

Adds a new value "v6v4" for IP_STACK to allow deployment of clusters with ipv6 as the primary address family.

@openshift-ci
openshift-ci Bot requested review from celebdor and sadasu January 26, 2023 16:44
@cybertron

Copy link
Copy Markdown
Contributor Author

/hold

Needs to be documented. Pushing the functional changes so we can collaborate on this feature.

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 26, 2023
@cybertron cybertron changed the title Add support for ipv6-primary dual stack OPNET-200: Add support for ipv6-primary dual stack Feb 3, 2023
@elfosardo

Copy link
Copy Markdown
Member

/retest

Adds a new value "v6v4" for IP_STACK to allow deployment of clusters
with ipv6 as the primary address family.
@cybertron

Copy link
Copy Markdown
Contributor Author

I believe the dual stack job is failing on a known SDN bug with dual stack. It looks like the cluster deployed fine though so it's unlikely to be related to this. I'm about to push a new revision with docs for the new value so it will reset the jobs anyway.

@cybertron

Copy link
Copy Markdown
Contributor Author

/hold cancel

Docs added. Note that this doesn't currently work due to lack of support in the rest of OpenShift, but we need this to develop and test that support anyway.

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Feb 7, 2023
@cybertron

Copy link
Copy Markdown
Contributor Author

/retest-required

@cybertron

Copy link
Copy Markdown
Contributor Author

/retest

Looks like dual stack is passing again. Let's see how it does with this.

@mkowalski

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Mar 6, 2023
@dtantsur

dtantsur commented Mar 6, 2023

Copy link
Copy Markdown
Member

/approve

@openshift-ci

openshift-ci Bot commented Mar 6, 2023

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dtantsur

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Mar 6, 2023
@openshift-merge-robot
openshift-merge-robot merged commit f3a3fbb into openshift-metal3:master Mar 6, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants