Skip to content

Doc: default release is now 4.12 - #1494

Merged
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
flavio-fernandes:4.12_doc
Dec 22, 2022
Merged

Doc: default release is now 4.12#1494
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
flavio-fernandes:4.12_doc

Conversation

@flavio-fernandes

Copy link
Copy Markdown
Contributor

TRIVIAL: Update docs and add a comment in common.sh as a reminder when default value changes.

This is a minor cont to #1459

Signed-off-by: Flavio Fernandes flaviof@redhat.com

TRIVIAL: Update docs and add a comment in common.sh as a reminder
when default value changes.

This is a minor cont to openshift-metal3#1459

Signed-off-by: Flavio Fernandes <flaviof@redhat.com>
@openshift-ci
openshift-ci Bot requested review from bfournie and sadasu December 16, 2022 18:54
@openshift-ci openshift-ci Bot added the needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. label Dec 16, 2022
@openshift-ci

openshift-ci Bot commented Dec 16, 2022

Copy link
Copy Markdown

Hi @flavio-fernandes. Thanks for your PR.

I'm waiting for a openshift-metal3 member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@flavio-fernandes

Copy link
Copy Markdown
Contributor Author

@dougsland PTAL :)

@derekhiggins

Copy link
Copy Markdown
Collaborator

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Dec 19, 2022
@elfosardo

Copy link
Copy Markdown
Member

/ok-to-test

@openshift-ci openshift-ci Bot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Dec 19, 2022
@cybertron

Copy link
Copy Markdown
Contributor

/approve

This is a doc-only change so there's no need to spend ci cycles retesting it.

@openshift-ci

openshift-ci Bot commented Dec 21, 2022

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cybertron

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 21, 2022
@derekhiggins

Copy link
Copy Markdown
Collaborator

/override e2e-metal-ipi-serial-ipv4

@openshift-ci

openshift-ci Bot commented Dec 22, 2022

Copy link
Copy Markdown

@derekhiggins: /override requires failed status contexts, check run or a prowjob name to operate on.
The following unknown contexts/checkruns were given:

  • e2e-metal-ipi-serial-ipv4

Only the following failed contexts/checkruns were expected:

  • ci/prow/e2e-agent-compact
  • ci/prow/e2e-agent-sno-ipv6
  • ci/prow/e2e-metal-ipi-bm
  • ci/prow/e2e-metal-ipi-bm-bond
  • ci/prow/e2e-metal-ipi-ovn-dualstack
  • ci/prow/e2e-metal-ipi-ovn-ipv6
  • ci/prow/e2e-metal-ipi-serial-ipv4
  • ci/prow/e2e-metal-ipi-serial-ovn-ipv6
  • ci/prow/e2e-metal-ipi-virtualmedia
  • ci/prow/images
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-compact
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-agent-sno-ipv6
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-metal-ipi-bm
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-metal-ipi-bm-bond
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-metal-ipi-ovn-dualstack
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-metal-ipi-ovn-ipv6
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-metal-ipi-serial-ipv4
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-metal-ipi-serial-ovn-ipv6
  • pull-ci-openshift-metal3-dev-scripts-master-e2e-metal-ipi-virtualmedia
  • pull-ci-openshift-metal3-dev-scripts-master-images
  • tide

If you are trying to override a checkrun that has a space in it, you must put a double quote on the context.

Details

In response to this:

/override e2e-metal-ipi-serial-ipv4

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@derekhiggins

Copy link
Copy Markdown
Collaborator

/override ci/prow/e2e-metal-ipi-serial-ipv4

@openshift-ci

openshift-ci Bot commented Dec 22, 2022

Copy link
Copy Markdown

@derekhiggins: Overrode contexts on behalf of derekhiggins: ci/prow/e2e-metal-ipi-serial-ipv4

Details

In response to this:

/override ci/prow/e2e-metal-ipi-serial-ipv4

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@derekhiggins

Copy link
Copy Markdown
Collaborator

/override ci/prow/e2e-metal-ipi-bm

@openshift-ci

openshift-ci Bot commented Dec 22, 2022

Copy link
Copy Markdown

@derekhiggins: Overrode contexts on behalf of derekhiggins: ci/prow/e2e-metal-ipi-bm

Details

In response to this:

/override ci/prow/e2e-metal-ipi-bm

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@derekhiggins

Copy link
Copy Markdown
Collaborator

/override

@openshift-ci

openshift-ci Bot commented Dec 22, 2022

Copy link
Copy Markdown

@derekhiggins: /override requires failed status contexts to operate on, but none was given

Details

In response to this:

/override

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@derekhiggins

Copy link
Copy Markdown
Collaborator

/override ci/prow/e2e-agent-sno-ipv6 ci/prow/e2e-metal-ipi-ovn-ipv6 ci/prow/e2e-metal-ipi-virtualmedia ci/prow/e2e-metal-ipi-ovn-dualstack

@openshift-ci

openshift-ci Bot commented Dec 22, 2022

Copy link
Copy Markdown

@derekhiggins: Overrode contexts on behalf of derekhiggins: ci/prow/e2e-agent-sno-ipv6, ci/prow/e2e-metal-ipi-ovn-dualstack, ci/prow/e2e-metal-ipi-ovn-ipv6, ci/prow/e2e-metal-ipi-virtualmedia

Details

In response to this:

/override ci/prow/e2e-agent-sno-ipv6 ci/prow/e2e-metal-ipi-ovn-ipv6 ci/prow/e2e-metal-ipi-virtualmedia ci/prow/e2e-metal-ipi-ovn-dualstack

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-merge-robot
openshift-merge-robot merged commit 32cdfce into openshift-metal3:master Dec 22, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants