Skip to content

Add padding to pass rates script output - #1487

Merged
openshift-merge-robot merged 2 commits into
openshift-metal3:masterfrom
honza:padding
Feb 15, 2023
Merged

Add padding to pass rates script output#1487
openshift-merge-robot merged 2 commits into
openshift-metal3:masterfrom
honza:padding

Conversation

@honza

@honza honza commented Dec 5, 2022

Copy link
Copy Markdown
Member

Sorry, my OCD just could not

@openshift-ci
openshift-ci Bot requested review from bfournie and flaper87 December 5, 2022 20:21
@honza

honza commented Dec 6, 2022

Copy link
Copy Markdown
Member Author

cc @derekhiggins

@dougsland

Copy link
Copy Markdown
Collaborator

looks sane to me.
/test e2e-metal-ipi-serial-ipv4

@derekhiggins

Copy link
Copy Markdown
Collaborator

/approve

@openshift-ci

openshift-ci Bot commented Dec 12, 2022

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: derekhiggins

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 12, 2022
Comment thread scripts/periodics_pass_rates.sh Outdated
@derekhiggins

Copy link
Copy Markdown
Collaborator

/lgtm
thanks

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 15, 2023
@openshift-merge-robot
openshift-merge-robot merged commit d8377aa into openshift-metal3:master Feb 15, 2023
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants