Skip to content

Add utility to list recent metal-ipi pass rates - #1482

Merged
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
derekhiggins:add-pass-rates
Nov 29, 2022
Merged

Add utility to list recent metal-ipi pass rates#1482
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
derekhiggins:add-pass-rates

Conversation

@derekhiggins

Copy link
Copy Markdown
Collaborator

No description provided.

@openshift-ci
openshift-ci Bot requested review from flaper87 and hardys November 28, 2022 10:40
function getJobSummary(){
JOB=$1
URL=https://prow.ci.openshift.org/job-history/gs/origin-ci-test/logs/$JOB
DATA=$(curl --silent $URL | grep allBuilds - | grep -o '\[.*\]')

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Do you want to handle failures?

@dtantsur

Copy link
Copy Markdown
Member

/approve

@openshift-ci

openshift-ci Bot commented Nov 28, 2022

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dtantsur

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Nov 28, 2022
@openshift-ci

openshift-ci Bot commented Nov 28, 2022

Copy link
Copy Markdown

@derekhiggins: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-metal-ipi-virtualmedia 3569e3d link false /test e2e-metal-ipi-virtualmedia

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@elfosardo

Copy link
Copy Markdown
Member

/retest

@elfosardo

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Nov 29, 2022
@openshift-merge-robot
openshift-merge-robot merged commit 915d147 into openshift-metal3:master Nov 29, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants