Skip to content

AGENT-283: Remove unwanted ClusterDeployment.Spec fields - #1425

Merged
openshift-ci[bot] merged 1 commit into
openshift-metal3:masterfrom
pawanpinjarkar:agent-remove-unwanted-fields-from-ClusterDeployment.spec
Jul 15, 2022
Merged

AGENT-283: Remove unwanted ClusterDeployment.Spec fields#1425
openshift-ci[bot] merged 1 commit into
openshift-metal3:masterfrom
pawanpinjarkar:agent-remove-unwanted-fields-from-ClusterDeployment.spec

Conversation

@pawanpinjarkar

@pawanpinjarkar pawanpinjarkar commented Jul 13, 2022

Copy link
Copy Markdown
Contributor

Remove the unwanted fields from the ClusterDeployment.Spec.

@openshift-ci
openshift-ci Bot requested review from celebdor and lranjbar July 13, 2022 17:13
@pawanpinjarkar

Copy link
Copy Markdown
Contributor Author

/cc @zaneb @rwsu

@openshift-ci
openshift-ci Bot requested review from rwsu and zaneb July 13, 2022 17:13
Comment thread agent/04_agent_configure.sh
Comment thread agent/04_agent_configure.sh
Comment thread agent/04_agent_configure.sh
@pawanpinjarkar
pawanpinjarkar force-pushed the agent-remove-unwanted-fields-from-ClusterDeployment.spec branch from 24f6b0e to cf1d40d Compare July 13, 2022 20:37
@pawanpinjarkar
pawanpinjarkar requested a review from zaneb July 13, 2022 20:38
@zaneb

zaneb commented Jul 13, 2022

Copy link
Copy Markdown
Member

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 13, 2022

@rwsu rwsu left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@openshift-ci

openshift-ci Bot commented Jul 15, 2022

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: rwsu

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 15, 2022
@openshift-ci
openshift-ci Bot merged commit 81674de into openshift-metal3:master Jul 15, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants