Skip to content

Support CentOS Stream 9 and RHEL 9 - #1367

Merged
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
elfosardo:support-centos-rhel-9
Jul 26, 2022
Merged

Support CentOS Stream 9 and RHEL 9#1367
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
elfosardo:support-centos-rhel-9

Conversation

@elfosardo

Copy link
Copy Markdown
Member

No description provided.

@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 23, 2022
@openshift-ci
openshift-ci Bot requested review from andfasano and hardys March 23, 2022 09:21
@elfosardo
elfosardo force-pushed the support-centos-rhel-9 branch 8 times, most recently from 30d6827 to 88c1511 Compare March 23, 2022 11:15
@elfosardo

Copy link
Copy Markdown
Member Author

/cc @derekhiggins

@openshift-ci
openshift-ci Bot requested a review from derekhiggins March 23, 2022 11:16
@elfosardo

Copy link
Copy Markdown
Member Author

/retest

2 similar comments
@elfosardo

Copy link
Copy Markdown
Member Author

/retest

@elfosardo

Copy link
Copy Markdown
Member Author

/retest

@elfosardo
elfosardo force-pushed the support-centos-rhel-9 branch from 88c1511 to aa5876a Compare March 31, 2022 13:03
@elfosardo elfosardo changed the title [WIP] Support CentOS Stream 9 and RHEL 9 Support CentOS Stream 9 and RHEL 9 Mar 31, 2022
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Mar 31, 2022
@elfosardo

Copy link
Copy Markdown
Member Author

/retest

@elfosardo
elfosardo force-pushed the support-centos-rhel-9 branch 2 times, most recently from 8474e63 to ba2fcef Compare March 31, 2022 14:35
@elfosardo

Copy link
Copy Markdown
Member Author

/retest

3 similar comments
@elfosardo

Copy link
Copy Markdown
Member Author

/retest

@elfosardo

Copy link
Copy Markdown
Member Author

/retest

@elfosardo

Copy link
Copy Markdown
Member Author

/retest

@elfosardo
elfosardo force-pushed the support-centos-rhel-9 branch from ba2fcef to 2ed0e1e Compare May 4, 2022 07:01
@openshift-ci openshift-ci Bot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label May 10, 2022
@elfosardo
elfosardo force-pushed the support-centos-rhel-9 branch from 2ed0e1e to a2734d2 Compare May 30, 2022 10:08
@openshift-ci openshift-ci Bot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label May 30, 2022
@elfosardo

Copy link
Copy Markdown
Member Author

also this is needed openshift/installer#6062

@elfosardo

Copy link
Copy Markdown
Member Author

this passes on current CI platform plus tested on CS8/CS9, it should be enough, just need to wait for the two mentioned PRs and of course reviews

@elfosardo

Copy link
Copy Markdown
Member Author

/test e2e-metal-ipi-ovn-ipv6

@elfosardo

Copy link
Copy Markdown
Member Author

/hold

@openshift-ci openshift-ci Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jul 19, 2022
@elfosardo
elfosardo force-pushed the support-centos-rhel-9 branch from 68be60f to b40ee39 Compare July 21, 2022 13:41
@elfosardo

Copy link
Copy Markdown
Member Author

/unhold

@openshift-ci openshift-ci Bot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jul 21, 2022
@elfosardo

Copy link
Copy Markdown
Member Author

/retest
this was working fine and it works fine locally, let's give it another try

@elfosardo
elfosardo force-pushed the support-centos-rhel-9 branch from b40ee39 to 084f7a9 Compare July 22, 2022 09:05
@elfosardo

Copy link
Copy Markdown
Member Author

/retest

@openshift-ci

openshift-ci Bot commented Jul 22, 2022

Copy link
Copy Markdown

@elfosardo: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-metal-ipi-proxy-ipv6 68be60f link false /test e2e-metal-ipi-proxy-ipv6
ci/prow/e2e-metal-ipi-proxy-ipv4 68be60f link false /test e2e-metal-ipi-proxy-ipv4

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@elfosardo

Copy link
Copy Markdown
Member Author

/retest

sudo pip3 install ansible=="${ANSIBLE_VERSION}"
# Also need the 3.9 version of netaddr for ansible.netcommon,
case $DISTRO in
"centos8"|"rhel8")

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Maybe short-circuit centos8 here to avoid checking for it 4 lines below?

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I will do both changes in a follow up, would rather avoid another CI pass

# install network-scripts package to be able to use legacy network commands
sudo dnf install -y network-scripts
if [[ $DISTRO == "centos8" ]]; then
echo "CentOS is not supported anymore. Please switch to CentOS Stream / RHEL / Rocky Linux"

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Fail here?

@dtantsur

Copy link
Copy Markdown
Member

/approve

@openshift-ci

openshift-ci Bot commented Jul 25, 2022

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dtantsur

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 25, 2022
@elfosardo

Copy link
Copy Markdown
Member Author

/cc @hardys @iurygregory

@openshift-ci

openshift-ci Bot commented Jul 26, 2022

Copy link
Copy Markdown

@elfosardo: GitHub didn't allow me to request PR reviews from the following users: iurygregory.

Note that only openshift-metal3 members and repo collaborators can review this PR, and authors cannot review their own PRs.

Details

In response to this:

/cc @hardys @iurygregory

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@derekhiggins

Copy link
Copy Markdown
Collaborator

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Jul 26, 2022
@jacob-anders

Copy link
Copy Markdown

/lgtm

@openshift-merge-robot
openshift-merge-robot merged commit 07bcf67 into openshift-metal3:master Jul 26, 2022
@flavio-fernandes

Copy link
Copy Markdown
Contributor

hi @elfosardo . I'm trying dev-scripts on centos9-stream and getting the following error:

# dnf group install -y --with-optional "Virtualization Host" && systemctl enable --now libvirtd && echo ok

...
level=debug msg=Terraform has been successfully initialized!
level=debug msg=[INFO] running Terraform command: /home/ff/dev-scripts/ocp/ostest/terraform/bin/terraform apply -no-color -auto-approve -input=false -var-file=/tmp/openshift-install-bootstrap-1219846892/terraform.tfvars.json -var-file=/tmp/openshift-install-bootstrap-1219846892/terraform.platform.auto.tfvars.json -lock=true -parallelism=10 -refresh=true
level=error
level=error msg=Error: failed to dial libvirt: dial unix /var/run/libvirt/libvirt-sock: connect: no such file or directory
level=error
level=error msg=  with provider["openshift/local/libvirt"],
level=error msg=  on main.tf line 1, in provider "libvirt":
level=error msg=   1: provider "libvirt" {
level=error
level=error msg=failed to fetch Cluster: failed to generate asset "Cluster": failure applying terraform for "bootstrap" stage: failed to create cluster: failed to apply Terraform: exit status 1
level=error
level=error msg=Error: failed to dial libvirt: dial unix /var/run/libvirt/libvirt-sock: connect: no such file or directory
level=error
level=error msg=  with provider["openshift/local/libvirt"],
level=error msg=  on main.tf line 1, in provider "libvirt":
level=error msg=   1: provider "libvirt" {
level=error
level=error
+(utils.sh:1): create_cluster(): auth_template_and_removetmp
+(utils.sh:672): auth_template_and_removetmp(): echo 4
+(utils.sh:673): auth_template_and_removetmp(): generate_auth_template
+(utils.sh:297): generate_auth_template(): set +x
Unable to connect to the server: dial tcp 192.168.111.5:6443: connect: no route to host
make: *** [Makefile:54: ocp_run] Error 1

Looking closely at my system, I can see that only the -ro file exist:

[ff@sdn-12 dev-scripts]$ sudo ls /var/run/libvirt/libvirt-soc*
/var/run/libvirt/libvirt-sock-ro

[root@sdn-12 ~]# virsh list --all
 Id   Name              State
----------------------------------
 -    ostest_master_0   shut off
 -    ostest_master_1   shut off
 -    ostest_master_2   shut off
 -    ostest_worker_0   shut off
 -    ostest_worker_1   shut off

[ff@sdn-12 dev-scripts]$ sudo systemctl status libvirtd
○ libvirtd.service - Virtualization daemon
     Loaded: loaded (/usr/lib/systemd/system/libvirtd.service; enabled; preset: disabled)
     Active: inactive (dead) since Wed 2023-01-11 15:27:44 EST; 8min ago
   Duration: 2min 109ms
TriggeredBy: ○ libvirtd-ro.socket
             ● libvirtd.socket
             ● libvirtd-admin.socket
             ○ libvirtd-tcp.socket
             ○ libvirtd-tls.socket
       Docs: man:libvirtd(8)
             https://libvirt.org
   Main PID: 10358 (code=exited, status=0/SUCCESS)
      Tasks: 2 (limit: 32768)
     Memory: 29.9M
        CPU: 1.246s
     CGroup: /system.slice/libvirtd.service
             ├─4691 /usr/sbin/dnsmasq --conf-file=/var/lib/libvirt/dnsmasq/default.conf --leasefile-ro --dhcp-script=/usr/libexec/libvirt_leaseshelper
             └─4692 /usr/sbin/dnsmasq --conf-file=/var/lib/libvirt/dnsmasq/default.conf --leasefile-ro --dhcp-script=/usr/libexec/libvirt_leaseshelper

Can you offer some pointers on whether this is an issue with terraform not using the -ro file or if I'm mssinf something that would make /var/run/libvirt/libvirt-sock available?

@flavio-fernandes

flavio-fernandes commented Jan 17, 2023

Copy link
Copy Markdown
Contributor

Update: I managed to work around this issue by not instaling libvirt explicitly!
Letting dev-scripts install libvirt for me must be different enough, such that /var/run/libvirt/libvirt-sock is created w/out problems. woot!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

6 participants