Skip to content

Use openshift_version function with is_lower_version - #1351

Merged
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
zaneb:openshift-version
Feb 8, 2022
Merged

Use openshift_version function with is_lower_version#1351
openshift-merge-robot merged 1 commit into
openshift-metal3:masterfrom
zaneb:openshift-version

Conversation

@zaneb

@zaneb zaneb commented Feb 8, 2022

Copy link
Copy Markdown
Member

The $OPENSHIFT_VERSION variable isn't defined in CI.

@openshift-ci
openshift-ci Bot requested review from hardys and markmc February 8, 2022 15:33
Comment thread common.sh Outdated
Comment thread run_ci.sh Outdated
Comment thread common.sh Outdated
Comment thread run_ci.sh Outdated
@zaneb
zaneb force-pushed the openshift-version branch from 900b6fe to 304946e Compare February 8, 2022 16:00
Comment thread common.sh Outdated
@hardys

hardys commented Feb 8, 2022

Copy link
Copy Markdown

/approve

Minor nit re the FIXME but not blocking on it :)

@openshift-ci

openshift-ci Bot commented Feb 8, 2022

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: hardys

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Feb 8, 2022
The $OPENSHIFT_VERSION variable doesn't work in CI any more.
@vikaslaad

Copy link
Copy Markdown

/retest-required

@zaneb
zaneb force-pushed the openshift-version branch from 304946e to 2799369 Compare February 8, 2022 17:34
@hardys

hardys commented Feb 8, 2022

Copy link
Copy Markdown

/lgtm

Merging to unblock nightly promotion

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Feb 8, 2022
@zaneb

zaneb commented Feb 8, 2022

Copy link
Copy Markdown
Member Author

Looking at the CI logs, we can already see this appears to be operating correctly:

++(ocp_install_env.sh:192): cluster_os_image(): is_lower_version 4.10 4.10
+++(common.sh:34): is_lower_version(): echo '4.10 4.10'
+++(common.sh:34): is_lower_version(): tr ' ' '\n'
+++(common.sh:34): is_lower_version(): sort -V
+++(common.sh:34): is_lower_version(): head -n1
++(common.sh:34): is_lower_version(): [[ 4.10 != 4.10 ]]
++(common.sh:37): is_lower_version(): return 1

I'd be comfortable overriding CI if needed to unblock nightlies.

@zaneb

zaneb commented Feb 8, 2022

Copy link
Copy Markdown
Member Author

Let's try this.
/override ci/prow/e2e-metal-ipi
/override ci/prow/e2e-metal-ipi-ovn-ipv6
/override ci/prow/e2e-metal-ipi-serial-ipv4
/override ci/prow/e2e-metal-ipi-serial-ovn-ipv6

@openshift-ci

openshift-ci Bot commented Feb 8, 2022

Copy link
Copy Markdown

@zaneb: Overrode contexts on behalf of zaneb: ci/prow/e2e-metal-ipi, ci/prow/e2e-metal-ipi-ovn-ipv6, ci/prow/e2e-metal-ipi-serial-ipv4, ci/prow/e2e-metal-ipi-serial-ovn-ipv6

Details

In response to this:

Let's try this.
/override ci/prow/e2e-metal-ipi
/override ci/prow/e2e-metal-ipi-ovn-ipv6
/override ci/prow/e2e-metal-ipi-serial-ipv4
/override ci/prow/e2e-metal-ipi-serial-ovn-ipv6

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@openshift-merge-robot
openshift-merge-robot merged commit 3f6fdcc into openshift-metal3:master Feb 8, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants