Skip to content

chore: refresh workflow_dispatch trigger - #94

Merged
rafabene merged 1 commit into
mainfrom
chore/refresh-workflow
Sep 11, 2026
Merged

rafabene merged 1 commit into
mainfrom
chore/refresh-workflow

Conversation

@rafabene

Copy link
Copy Markdown
Member

Minor workflow configuration refresh to enable manual dispatch.

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign kuudori for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@rafabene
rafabene merged commit 53aee8c into main Sep 11, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant