Skip to content

product: Drop ose-hello-openshift-container mapping - #6829

Merged
openshift-merge-bot[bot] merged 1 commit into
openshift-eng:mainfrom
wking:drop-hello-openshift-mapping
May 28, 2025
Merged

openshift-merge-bot[bot] merged 1 commit into
openshift-eng:mainfrom
wking:drop-hello-openshift-mapping

Conversation

@wking

@wking wking commented May 27, 2025

Copy link
Copy Markdown
Contributor

ART hasn't been building these images since 4.8's a6286e7, with the router/ingress folks pivoting away in openshift/cluster-ingress-operator@8332af2f56 (openshift/cluster-ingress-operator#561). Samples had been installing a hello-openshift ImageStream, but they also got away from that in 2021 with
openshift/cluster-samples-operator@d94ad97e49 (openshift/cluster-samples-operator#380). Origin hasn't entirely gotten out of the hello-openshift business yet, but regardless of whether they still have an interest, neither ART nor ingress is involved anymore, and it's been a long time since 4.7 went end-of-life:

$ curl -s 'https://access.redhat.com/product-life-cycles/api/v1/products?name=Openshift+Container+Platform+4' | jq -r '.data[].versions[] | select(.name == "4.7").phases[] | .date + " " + .name' 2021-02-24T00:00:00.000Z General availability
2021-10-27T00:00:00.000Z Full support
2022-08-24T00:00:00.000Z Maintenance support
N/A Extended update support
N/A Extended update support Term 2
N/A Extended life phase

So I'm dropping those lines from the mapping.

ART hasn't been building these images since 4.8's a6286e7 (rm
hello-openshift, 2021-03-29), with the router/ingress folks pivoting
away in openshift/cluster-ingress-operator@8332af2f56 (NE 484: Use
ingress-operator subcommand instead of hello-openshift for canary
server, 2021-03-18, openshift/cluster-ingress-operator#561).  Samples
had been installing a hello-openshift ImageStream, but they also got
away from that in 2021 with
openshift/cluster-samples-operator@d94ad97e49 (delete hello-openshift
in payload imagestream via CVO annotation, 2021-06-24,
openshift/cluster-samples-operator#380).  Origin hasn't entirely
gotten out of the hello-openshift business yet [1], but regardless of
whether they still have an interest, neither ART nor ingress is
involved anymore, and it's been a long time since 4.7 went
end-of-life:

$ curl -s 'https://access.redhat.com/product-life-cycles/api/v1/products?name=Openshift+Container+Platform+4' | jq -r '.data[].versions[] | select(.name == "4.7").phases[] | .date + " " + .name'
2021-02-24T00:00:00.000Z General availability
2021-10-27T00:00:00.000Z Full support
2022-08-24T00:00:00.000Z Maintenance support
N/A Extended update support
N/A Extended update support Term 2
N/A Extended life phase

So I'm dropping those lines from the mapping.

[1]: https://github.com/openshift/origin/tree/94a4d2a6f202f64f89a1b747b44484c358f8299d/examples/hello-openshift

@joepvd joepvd left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks!
/lgtm
/approve

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label May 28, 2025
@openshift-ci

openshift-ci Bot commented May 28, 2025

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: joepvd

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label May 28, 2025
@openshift-merge-bot
openshift-merge-bot Bot merged commit f9bc4d8 into openshift-eng:main May 28, 2025
@wking
wking deleted the drop-hello-openshift-mapping branch May 28, 2025 16:23
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants