Skip to content

fix issue in WaiterWaiter caused by race condition - #4829

Merged
frank-dong-ms-zz merged 5 commits into
dotnet:masterfrom
frank-dong-ms-zz:SdcaBinary-issue
Feb 13, 2020
Merged

frank-dong-ms-zz merged 5 commits into
dotnet:masterfrom
frank-dong-ms-zz:SdcaBinary-issue

Conversation

@frank-dong-ms-zz

@frank-dong-ms-zz frank-dong-ms-zz commented Feb 12, 2020 •

Copy link
Copy Markdown
Contributor
  1. Fix issue in WaiterWaiter caused by race condition.
  2. Re-enable affected tests as below:
    EntryPointEvaluateRegression
    EntryPointSdcaBinary
    EntryPointSDCARegression
    EntryPointChainedCrossValMacros
    TestCrossValidationMacroWithStratification
    TestOvaMacro
    TrainAndPredictOnIris
    CrossValidationIris
    LinearClassifierTest
    SavePipeTrainAndScoreFccTransformStr
    BinaryClassifierLogisticRegressionNonNegativeTest
    InpsectLinearModelParameters
    TrainRegressionModel
    CustomTransformer
    TrainAndPredictIrisModelTest
    LightGbmBinaryClassificationOnnxConversionTest

@frank-dong-ms-zz
frank-dong-ms-zz marked this pull request as ready for review February 13, 2020 00:02
@frank-dong-ms-zz
frank-dong-ms-zz requested a review from a team as a code owner February 13, 2020 00:02
@frank-dong-ms-zz frank-dong-ms-zz changed the title fix issue in WaiterWaiter caused by background thread update fix issue in WaiterWaiter caused by race condition Feb 13, 2020
Comment thread src/Microsoft.ML.Data/DataView/CacheDataView.cs
Comment thread src/Microsoft.ML.Data/DataView/CacheDataView.cs Outdated
Comment thread src/Microsoft.ML.Data/DataView/CacheDataView.cs Outdated
Comment thread src/Microsoft.ML.Data/DataView/CacheDataView.cs Outdated
Comment thread src/Microsoft.ML.Data/DataView/CacheDataView.cs Outdated

@harishsk harishsk left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

:shipit:

@frank-dong-ms-zz
frank-dong-ms-zz merged commit 6cd6081 into dotnet:master Feb 13, 2020
@frank-dong-ms-zz
frank-dong-ms-zz deleted the SdcaBinary-issue branch April 15, 2020 20:36
@ghost ghost locked as resolved and limited conversation to collaborators Mar 19, 2022
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants