Skip to content

Adds CODEOWNERS file - #1

Merged
Samruddhi Khandale (samruddhikhandale) merged 1 commit into
mainfrom
samruddhikhandale-add-codeowners
Jul 28, 2022
Merged

Samruddhi Khandale (samruddhikhandale) merged 1 commit into
mainfrom
samruddhikhandale-add-codeowners

Conversation

@samruddhikhandale

Copy link
Copy Markdown
Member

Adding @devcontainers/maintainers as CODEOWNER for all the files in this repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants