feat(miniflare): add unsafeRegisterWorker option to restrict worker registration - #10411
Closed
edmundhung wants to merge 2 commits into
Closed
edmundhung wants to merge 2 commits into
edmundhung wants to merge 2 commits into
Conversation
🦋 Changeset detectedLatest commit: 049514c The changes in this PR will be included in the next version bump. This PR includes changesets to release 5 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
create-cloudflare
@cloudflare/kv-asset-handler
miniflare
@cloudflare/pages-shared
@cloudflare/unenv-preset
@cloudflare/vite-plugin
@cloudflare/vitest-pool-workers
@cloudflare/workers-editor-shared
wrangler
commit: |
edmundhung
force-pushed
the
edmundhung/restrict-registry-registration
branch
3 times, most recently
from
August 21, 2025 12:56
e4578b2 to
774e6ac
Compare
edmundhung
force-pushed
the
edmundhung/restrict-registry-registration
branch
from
September 1, 2025 10:55
774e6ac to
180f01a
Compare
2 of 5 tasks
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Fixes n/a.
We were registering all workers to the dev registry, including external workers (e.g. for the vectorize binding). This restricts it to user worker only.